CATROID-1106: GlideToPositionBrick - #5239
Conversation
| @@ -0,0 +1,75 @@ | |||
| <?xml version="1.0" encoding="utf-8"?> | |||
|
|
|||
| <!-- | |||
There was a problem hiding this comment.
detekt found more than 20 potential problems in the proposed changes. Check the Files changed tab for more details.
|
please fix all warnings first and have a look at the Unit Test Report, there are failing testcases. Not sure if they are related to these changes. Then i can do a review :) |
A new Brick (Feature) is created. The Brick has 3 options: -random position - the sprite glides to a random position -touch position - the player has to touch a position where the sprite then goes -other sprite position - the sprite glides and at the moment follows a other sprite till the other sprite stops. - this part is not finally discussed: may we change it so that the sprite goes to the startposition from the other sprite, if it moves too TestGlideToRandomPositionDestination Test; flact:red TestGlideToRandomPositionDestination flact:green; createGlidetoRandomPositionAction; Refactoring and knew Test: GlideBehavior, bove Tests are green show Brick in Bricklist with the correct attributs. Attributes don't work yet. changed all glide_to_.... to glide_to_position_... so to better distinguish between. last commit before PairProgramming Session with Thomas glideToTouchPosition ready, brick test by hand and the tests are grenn refactor the ActionTest: canceled all deprecated lines; correct misspellings; add GlideToOtherSpritePosition refactor GlideToRandomPositionAction.kt Refactoring all files that changed in the ticket Catroid-1106 Refactoring all files that changed in the ticket Catroid-1106
2ba8a8c to
23b5c77
Compare
…sts suddently failed and we do not have a koin base class
…e problem with the testpipeline
…ction for the GlideToPosition actions. Warnings of deprecated stays in the ActionFactory
…ction for the GlideToPosition actions. Warnings of deprecated stays in the ActionFactory
|
|
@hannesweilharter please start with CodeReview. |



https://catrobat.atlassian.net/browse/CATROID-1106
A new Brick (Feature) is created.
Brick Behavior:
The Brick has 3 options:
-random position
- the sprite glides to a random position
-touch position
- the player has to touch a position where the sprite then goes
-other sprite position
- the sprite glides (and at the moment follows) a other sprite till the other sprite stops.
- this part is not finally discussed: may we change it so that the sprite goes to the startposition from the other sprite and don't follow at all.
Program changes:
create 3 Actions and 3 Actionstest, GlideToPositionBrick, brick_glide_to_position
changes in BrickValue, ActionFactory
Commitlist during coding:
TestGlideToRandomPositionDestination Test; flact:red
TestGlideToRandomPositionDestination flact:green; createGlidetoRandomPositionAction;
Refactoring and knew Test: GlideBehavior, bove Tests are green
show Brick in Bricklist with the correct attributs. Attributes don't work yet.
changed all glide_to_.... to glide_to_position_... so to better distinguish between.
last commit before PairProgramming Session with Thomas
glideToTouchPosition ready, brick test by hand and the tests are grenn
refactor the ActionTest: canceled all deprecated lines; correct misspellings;
add GlideToOtherSpritePosition
refactor GlideToRandomPositionAction.kt
Refactoring all files that changed in the ticket Catroid-1106
Your checklist for this pull request
Please review the contributing guidelines and wiki pages of this repository.