Skip to content

Add CODEOWNERS file for the repo - #1165

Merged
jesusmpc merged 2 commits into
mainfrom
internal/GH-1164-add-codeowners-file-for-the
Sep 3, 2026
Merged

Add CODEOWNERS file for the repo#1165
jesusmpc merged 2 commits into
mainfrom
internal/GH-1164-add-codeowners-file-for-the

Conversation

@jesusmpc

@jesusmpc jesusmpc commented Sep 3, 2026

Copy link
Copy Markdown
Contributor

Closes #1164

@jesusmpc jesusmpc self-assigned this Sep 3, 2026
@jesusmpc jesusmpc added the skip-release Skips the release creation label Sep 3, 2026
@jesusmpc
jesusmpc marked this pull request as ready for review September 3, 2026 13:49
@jesusmpc
jesusmpc merged commit 6437db9 into main Sep 3, 2026
15 checks passed
@jesusmpc
jesusmpc deleted the internal/GH-1164-add-codeowners-file-for-the branch September 3, 2026 13:53
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

skip-release Skips the release creation

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Add CODEOWNERS file for the repo

1 participant