Skip to content

Bump webpack-dev-middleware from 7.4.2 to 7.4.6 - #360

Merged
github-actions[bot] merged 1 commit into
mainfrom
dependabot/npm_and_yarn/webpack-dev-middleware-7.4.6
Oct 1, 2026
Merged

github-actions[bot] merged 1 commit into
mainfrom
dependabot/npm_and_yarn/webpack-dev-middleware-7.4.6

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 1, 2026

Copy link
Copy Markdown
Contributor

Bumps webpack-dev-middleware from 7.4.2 to 7.4.6.

Release notes

Sourced from webpack-dev-middleware's releases.

v7.4.6

7.4.6 (2026-09-03)

Bug Fixes

  • reject requests that resolve outside the output directory

v7.4.5

7.4.5 (2025-09-24)

Bug Fixes

v7.4.4

7.4.4 (2025-09-23)

Bug Fixes

v7.4.3

7.4.3 (2025-09-05)

Bug Fixes

  • do not call the next middleware for 304 responses (#2155) (c26a326)
  • do not call the next middleware when request is finished or errored (#2156) (116c680)
Changelog

Sourced from webpack-dev-middleware's changelog.

7.4.6 (2026-09-03)

Bug Fixes

  • reject requests that resolve outside the output directory

7.4.5 (2025-09-24)

Bug Fixes

7.4.4 (2025-09-23)

Bug Fixes

7.4.3 (2025-09-05)

Bug Fixes

  • do not call the next middleware for 304 responses (#2155) (c26a326)
  • do not call the next middleware when request is finished or errored (#2156) (116c680)
Commits
  • 70b3c33 chore(release): 7.4.6
  • cefccba Merge commit from fork
  • ec97e06 chore(release): 7.4.5
  • c9a0e68 fix: unpin memfs (#2176)
  • 0c87e95 chore(deps-dev): bump the dependencies group across 1 directory with 6 update...
  • 8f40f5f chore(release): 7.4.4
  • 044d691 fix: pin memfs version (#2174)
  • 253c7e1 chore(deps-dev): bump the dependencies group with 3 updates (#2170)
  • 2535262 chore(deps): bump the dependencies group across 1 directory with 5 updates (#...
  • 6b7531e chore: remove CODEOWNERS, too spammy (#2167)
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)
    You can disable automated security fix PRs for this repo from the Security Alerts page.

Bumps [webpack-dev-middleware](https://github.com/webpack/webpack-dev-middleware) from 7.4.2 to 7.4.6.
- [Release notes](https://github.com/webpack/webpack-dev-middleware/releases)
- [Changelog](https://github.com/webpack/webpack-dev-middleware/blob/v7.4.6/CHANGELOG.md)
- [Commits](webpack/webpack-dev-middleware@v7.4.2...v7.4.6)

---
updated-dependencies:
- dependency-name: webpack-dev-middleware
  dependency-version: 7.4.6
  dependency-type: indirect
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code labels Oct 1, 2026
@github-actions
github-actions Bot enabled auto-merge (squash) October 1, 2026 16:00
@github-actions
github-actions Bot merged commit 28bd042 into main Oct 1, 2026
3 checks passed
@github-actions
github-actions Bot deleted the dependabot/npm_and_yarn/webpack-dev-middleware-7.4.6 branch October 1, 2026 16:25
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants