Skip to content

Update module go.uber.org/atomic to v1.12.0 - #196

Merged
gardener-prow[bot] merged 1 commit into
mainfrom
renovate/go.uber.org-atomic-1.x
Sep 17, 2026
Merged

gardener-prow[bot] merged 1 commit into
mainfrom
renovate/go.uber.org-atomic-1.x

Conversation

@gardener-ci-robot

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
go.uber.org/atomic v1.11.0v1.12.0 age confidence

Release Notes

uber-go/atomic (go.uber.org/atomic)

v1.12.0

Compare Source

Added
  • Add MarshalJSON and UnmarshalJSON method to atomic.Pointer[T] type
    allowing users to use pointer with json.
  • Add MarshalJSON and UnmarshalJSON methods to atomic.Time so it
    encodes and decodes the wrapped time.Time instead of an empty struct.
Updated

Full Changelog: uber-go/atomic@v1.11.0...v1.12.0


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

Signed-off-by: gardener-ci-robot <gardener.ci.robot@gmail.com>
@gardener-ci-robot
gardener-ci-robot requested a review from a team as a code owner September 17, 2026 22:36
@gardener-ci-robot gardener-ci-robot added the kind/enhancement Enhancement, improvement, extension label Sep 17, 2026
@gardener-ci-robot

Copy link
Copy Markdown
Contributor Author

/label skip-review

@gardener-prow

gardener-prow Bot commented Sep 17, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign anluerz for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 17, 2026
@gardener-prow
gardener-prow Bot merged commit 80b042b into main Sep 17, 2026
25 checks passed
@gardener-ci-robot
gardener-ci-robot deleted the renovate/go.uber.org-atomic-1.x branch September 17, 2026 23:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/enhancement Enhancement, improvement, extension size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant