Skip to content

Update kubernetes monorepo to v0.37.1 - #197

Merged
gardener-prow[bot] merged 1 commit into
mainfrom
renovate/patch-kubernetes-monorepo
Sep 24, 2026
Merged

gardener-prow[bot] merged 1 commit into
mainfrom
renovate/patch-kubernetes-monorepo

Conversation

@gardener-ci-robot

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
k8s.io/api v0.37.0 → v0.37.1 age confidence
k8s.io/apimachinery v0.37.0 → v0.37.1 age confidence
k8s.io/client-go v0.37.0 → v0.37.1 age confidence

Release Notes

kubernetes/api (k8s.io/api)

v0.37.1

Compare Source

kubernetes/apimachinery (k8s.io/apimachinery)

v0.37.1

Compare Source

kubernetes/client-go (k8s.io/client-go)

v0.37.1

Compare Source


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

Signed-off-by: gardener-ci-robot <gardener.ci.robot@gmail.com>
@gardener-ci-robot
gardener-ci-robot requested a review from a team as a code owner September 24, 2026 02:35
@gardener-ci-robot gardener-ci-robot added the kind/enhancement Enhancement, improvement, extension label Sep 24, 2026
@gardener-ci-robot

Copy link
Copy Markdown
Contributor Author

/label skip-review

@gardener-prow gardener-prow Bot added the skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. label Sep 24, 2026
@gardener-prow

gardener-prow Bot commented Sep 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign scheererj for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 24, 2026
@gardener-prow
gardener-prow Bot merged commit ccc6a7a into main Sep 24, 2026
24 checks passed
@gardener-ci-robot
gardener-ci-robot deleted the renovate/patch-kubernetes-monorepo branch September 24, 2026 03:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/enhancement Enhancement, improvement, extension size/S Denotes a PR that changes 10-29 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant