fix(deps): update gomod dependencies - #1029
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
Conversation
ℹ️ Artifact update noticeFile name: api/go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
File name: go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
|
|
Hi @red-hat-konflux-kflux-prd-rh03[bot]. Thanks for your PR. I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
df182a8 to
c1fde98
Compare
|
New changes are detected. LGTM label has been removed. |
6c83fe2 to
c51502d
Compare
|
PR needs rebase. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
b8a56ce to
1cf6197
Compare
0bc6e09 to
c191900
Compare
8274849 to
3fa011b
Compare
4ffeacb to
6574044
Compare
|
/retest |
|
@red-hat-konflux-kflux-prd-rh03[bot]: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
009324b to
1c3692a
Compare
4b9c91f to
5697419
Compare
e06f612 to
7adcb81
Compare
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: Repository YAML (base), Central YAML (inherited) Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
Signed-off-by: red-hat-konflux-kflux-prd-rh03 <206760901+red-hat-konflux-kflux-prd-rh03[bot]@users.noreply.github.com>
|
[APPROVALNOTIFIER] This PR is APPROVED Approval requirements bypassed by manually added approval. This pull-request has been approved by: red-hat-konflux-kflux-prd-rh03[bot] The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
This PR contains the following updates:
v2.6.1+incompatible→v2.7.0+incompatiblev1.41.7→v1.45.1v1.7.10→v1.7.20v1.32.18→v1.33.1v1.19.17→v1.20.1v1.18.23→v1.19.1v1.4.23→v1.5.1v2.7.23→v2.8.1v1.4.24→v1.5.1v1.31.0→v1.38.1v1.304.0→v1.325.1v1.53.10→v1.61.1v1.13.9→v1.13.19v1.9.16→v1.11.1v1.13.23→v1.14.1v1.19.23→v1.20.1v1.51.4→v1.56.1v1.62.7→v1.67.1v1.102.0→v1.109.1v1.35.0→v1.39.1v1.0.11→v1.7.1v1.30.17→v1.35.1v1.36.0→v1.40.1v1.42.1→v1.47.1v1.31.23→v1.36.1v1.25.1→v1.28.1v5.6.0→v5.9.11v1.3.0→v1.4.4v1.2.0→v1.4.4v0.23.1→v0.24.0v0.21.5→v0.21.6v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.1v0.26.0→v0.29.141bb18b→2c02b82v0.5.7-v3refs→v0.7.1v0.6.0→v0.7.0v1.1.0→v1.2.0v1.1.2→v1.6.0v2.6.0→v2.32.1v1.24.1→v1.43.0a83e6f8→c7d4aa1v0.11.0→v0.19.0v0.55.0→v0.93.1v1.14.0→v1.24.1v0.3.0→v0.6.2v0.37.0→v0.70.1v0.8.0→v0.21.1v0.22.0v0.4.0→v0.6.0v1.10.0→v1.11.0v1.26.0→v1.28.0v3.0.4→v3.0.5v0.55.0→v0.58.0v0.38.0→v0.58.0v0.27.0→v0.36.0v0.45.0→v0.47.0v0.43.0→v0.45.0v0.37.0→v0.41.0v0.23.0→v0.41.0v0.3.0→v0.15.0v2.2.0→v2.5.0v1.33.0→v1.36.12v0.26.1→v0.37.0v0.24.0→v0.37.0v0.26.1→v0.37.0v0.26.1→v0.37.0v0.24.0→v0.37.0v0.26.1→v0.37.0v0.26.1→v0.37.0v2.80.1→v2.140.0v2.60.1→v2.140.0172d655→be32def3ee0da9→be32def99ec85e→cf1189d3a6ce19→cf1189dv0.14.6→v0.24.1v0.12.1→v0.24.1f223a00→2d320269f7c6b3→2d32026v4.2.3→v4.7.0v4.2.1→v4.7.0v1.3.0→v1.6.0Warning
Some dependencies could not be looked up. Check the Dependency Dashboard for more information.
Release Notes
avast/retry-go (github.com/avast/retry-go)
v2.7.0+incompatibleCompare Source
aws/aws-sdk-go-v2 (github.com/aws/aws-sdk-go-v2)
v1.45.1Compare Source
v1.45.0Compare Source
General Highlights
Module Highlights
github.com/aws/aws-sdk-go-v2/service/accessanalyzer: v1.50.0github.com/aws/aws-sdk-go-v2/service/account: v1.33.0github.com/aws/aws-sdk-go-v2/service/acm: v1.42.0github.com/aws/aws-sdk-go-v2/service/acmpca: v1.48.0github.com/aws/aws-sdk-go-v2/service/aiops: v1.8.0github.com/aws/aws-sdk-go-v2/service/amp: v1.45.0github.com/aws/aws-sdk-go-v2/service/amplify: v1.40.0github.com/aws/aws-sdk-go-v2/service/amplifybackend: v1.34.0github.com/aws/aws-sdk-go-v2/service/amplifyuibuilder: v1.30.0github.com/aws/aws-sdk-go-v2/service/apigateway: v1.41.0github.com/aws/aws-sdk-go-v2/service/apigatewaymanagementapi: v1.31.0github.com/aws/aws-sdk-go-v2/service/apigatewayv2: v1.36.0github.com/aws/aws-sdk-go-v2/service/appconfig: v1.46.0github.com/aws/aws-sdk-go-v2/service/appconfigdata: v1.25.0github.com/aws/aws-sdk-go-v2/service/appfabric: v1.18.0github.com/aws/aws-sdk-go-v2/service/appflow: v1.53.0github.com/aws/aws-sdk-go-v2/service/appintegrations: v1.39.0github.com/aws/aws-sdk-go-v2/service/applicationautoscaling: v1.44.0github.com/aws/aws-sdk-go-v2/service/applicationcostprofiler: v1.29.0github.com/aws/aws-sdk-go-v2/service/applicationdiscoveryservice: v1.37.0github.com/aws/aws-sdk-go-v2/service/applicationinsights: v1.36.0github.com/aws/aws-sdk-go-v2/service/applicationsignals: v1.24.0github.com/aws/aws-sdk-go-v2/service/appmesh: v1.37.0github.com/aws/aws-sdk-go-v2/service/apprunner: v1.41.0github.com/aws/aws-sdk-go-v2/service/appstream: v1.62.0github.com/aws/aws-sdk-go-v2/service/appsync: v1.55.0github.com/aws/aws-sdk-go-v2/service/arcregionswitch: v1.10.0github.com/aws/aws-sdk-go-v2/service/arczonalshift: v1.24.0github.com/aws/aws-sdk-go-v2/service/artifact: v1.18.0github.com/aws/aws-sdk-go-v2/service/athena: v1.59.0github.com/aws/aws-sdk-go-v2/service/auditmanager: v1.48.0github.com/aws/aws-sdk-go-v2/service/autoscaling: v1.69.0github.com/aws/aws-sdk-go-v2/service/autoscalingplans: v1.32.0github.com/aws/aws-sdk-go-v2/service/b2bi: v1.0.0-preview.113github.com/aws/aws-sdk-go-v2/service/backup: v1.58.0github.com/aws/aws-sdk-go-v2/service/backupgateway: v1.28.0github.com/aws/aws-sdk-go-v2/service/backupsearch: v1.8.0github.com/aws/aws-sdk-go-v2/service/batch: v1.67.0github.com/aws/aws-sdk-go-v2/service/bcmdashboards: v1.4.0github.com/aws/aws-sdk-go-v2/service/bcmdataexports: v1.17.0github.com/aws/aws-sdk-go-v2/service/bcmpricingcalculator: v1.12.0github.com/aws/aws-sdk-go-v2/service/bcmrecommendedactions: v1.3.0github.com/aws/aws-sdk-go-v2/service/bedrock: v1.65.0github.com/aws/aws-sdk-go-v2/service/bedrockagent: v1.57.0github.com/aws/aws-sdk-go-v2/service/bedrockagentcore: v1.33.0github.com/aws/aws-sdk-go-v2/service/bedrockagentcorecontrol: v1.46.0github.com/aws/aws-sdk-go-v2/service/bedrockagentruntime: v1.54.0github.com/aws/aws-sdk-go-v2/service/bedrockdataautomation: v1.17.0github.com/aws/aws-sdk-go-v2/service/bedrockdataautomationruntime: v1.12.0github.com/aws/aws-sdk-go-v2/service/bedrockruntime: v1.55.0github.com/aws/aws-sdk-go-v2/service/billing: v1.12.0github.com/aws/aws-sdk-go-v2/service/billingconductor: v1.31.0github.com/aws/aws-sdk-go-v2/service/braket: v1.42.0github.com/aws/aws-sdk-go-v2/service/budgets: v1.45.0github.com/aws/aws-sdk-go-v2/service/chatbot: v1.16.0github.com/aws/aws-sdk-go-v2/service/chime: v1.43.0github.com/aws/aws-sdk-go-v2/service/chimesdkidentity: v1.29.0github.com/aws/aws-sdk-go-v2/service/chimesdkmediapipelines: v1.28.0github.com/aws/aws-sdk-go-v2/service/chimesdkmeetings: v1.35.0github.com/aws/aws-sdk-go-v2/service/chimesdkmessaging: v1.34.0github.com/aws/aws-sdk-go-v2/service/chimesdkvoice: v1.30.0github.com/aws/aws-sdk-go-v2/service/cleanrooms: v1.47.0github.com/aws/aws-sdk-go-v2/service/cleanroomsml: v1.25.0github.com/aws/aws-sdk-go-v2/service/cloud9: v1.35.0github.com/aws/aws-sdk-go-v2/service/cloudcontrol: v1.31.0github.com/aws/aws-sdk-go-v2/service/clouddirectory: v1.32.0github.com/aws/aws-sdk-go-v2/service/cloudformation: v1.74.0github.com/aws/aws-sdk-go-v2/service/cloudfront: v1.66.0github.com/aws/aws-sdk-go-v2/service/cloudfrontkeyvaluestore: v1.14.0github.com/aws/aws-sdk-go-v2/service/cloudhsm: v1.31.0github.com/aws/aws-sdk-go-v2/service/cloudhsmv2: v1.36.0github.com/aws/aws-sdk-go-v2/service/cloudsearch: v1.34.0github.com/aws/aws-sdk-go-v2/service/cloudsearchdomain: v1.30.0github.com/aws/aws-sdk-go-v2/service/cloudtrail: v1.57.0github.com/aws/aws-sdk-go-v2/service/cloudtraildata: v1.19.0github.com/aws/aws-sdk-go-v2/service/cloudwatch: v1.62.0github.com/aws/aws-sdk-go-v2/service/cloudwatchevents: v1.34.0github.com/aws/aws-sdk-go-v2/service/cloudwatchlogs: v1.79.0github.com/aws/aws-sdk-go-v2/service/codeartifact: v1.40.0github.com/aws/aws-sdk-go-v2/service/codebuild: v1.71.0github.com/aws/aws-sdk-go-v2/service/codecatalyst: v1.23.0github.com/aws/aws-sdk-go-v2/service/codecommit: v1.35.0github.com/aws/aws-sdk-go-v2/service/codeconnections: v1.12.0github.com/aws/aws-sdk-go-v2/service/codedeploy: v1.37.0github.com/aws/aws-sdk-go-v2/service/codeguruprofiler: v1.31.0github.com/aws/aws-sdk-go-v2/service/codegurureviewer: v1.36.0github.com/aws/aws-sdk-go-v2/service/codegurusecurity: v1.18.0github.com/aws/aws-sdk-go-v2/service/codepipeline: v1.48.0github.com/aws/aws-sdk-go-v2/service/codestarconnections: v1.37.0github.com/aws/aws-sdk-go-v2/service/codestarnotifications: v1.33.0github.com/aws/aws-sdk-go-v2/service/cognitoidentity: v1.35.0github.com/aws/aws-sdk-go-v2/service/cognitoidentityprovider: v1.64.0github.com/aws/aws-sdk-go-v2/service/cognitosync: v1.31.0github.com/aws/aws-sdk-go-v2/service/comprehend: v1.42.0github.com/aws/aws-sdk-go-v2/service/comprehendmedical: v1.34.0github.com/aws/aws-sdk-go-v2/service/computeoptimizer: v1.55.0github.com/aws/aws-sdk-go-v2/service/computeoptimizerautomation: v1.4.0github.com/aws/aws-sdk-go-v2/service/configservice: v1.66.0github.com/aws/aws-sdk-go-v2/service/connect: v1.180.0github.com/aws/aws-sdk-go-v2/service/connectcampaigns: v1.22.0github.com/aws/aws-sdk-go-v2/service/connectcampaignsv2: v1.16.0github.com/aws/aws-sdk-go-v2/service/connectcases: v1.43.0github.com/aws/aws-sdk-go-v2/service/connectcontactlens: v1.35.0github.com/aws/aws-sdk-go-v2/service/connecthealth: v1.5.0github.com/aws/aws-sdk-go-v2/service/connectparticipant: v1.38.0github.com/aws/aws-sdk-go-v2/service/controlcatalog: v1.17.0github.com/aws/aws-sdk-go-v2/service/controltower: v1.30.0github.com/aws/aws-sdk-go-v2/service/costandusagereportservice: v1.36.0github.com/aws/aws-sdk-go-v2/service/costexplorer: v1.66.0github.com/aws/aws-sdk-go-v2/service/costoptimizationhub: v1.25.0github.com/aws/aws-sdk-go-v2/service/customerprofiles: [v1Configuration
📅 Schedule: (in timezone UTC)
* 2-4 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.