Skip to content

ROSAENG-67043: Add auto-release pipeline for tag pushes - #888

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
joshbranham:auto-release-pipeline
Sep 15, 2026
Merged

openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
joshbranham:auto-release-pipeline

Conversation

@joshbranham

Copy link
Copy Markdown
Contributor

This adds a new image-tag-push pipeline that will run and make a snapshot on tags on this repo. We can then make an accompanying MR to Konflux to respond and auto-release from this new pipeline. We set the existing image-push to not auto release (we can also delete that once we know this works)

@joshbranham joshbranham self-assigned this Sep 14, 2026
@openshift-ci
openshift-ci Bot requested review from bmeng and dustman9000 September 14, 2026 22:29
@coderabbitai

coderabbitai Bot commented Sep 14, 2026

Copy link
Copy Markdown

Comment @coderabbitai help to get the list of available commands.

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 14, 2026
@joshbranham joshbranham changed the title Add auto-release pipeline for tag pushes ROSAENG-67043: Add auto-release pipeline for tag pushes Sep 14, 2026
@openshift-ci-robot

openshift-ci-robot commented Sep 14, 2026

Copy link
Copy Markdown

@joshbranham: This pull request references ROSAENG-67043 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "5.1.0" version, but no target version was set.

Details

In response to this:

This adds a new image-tag-push pipeline that will run and make a snapshot on tags on this repo. We can then make an accompanying MR to Konflux to respond and auto-release from this new pipeline. We set the existing image-push to not auto release (we can also delete that once we know this works)

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 14, 2026
@joshbranham

Copy link
Copy Markdown
Contributor Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 14, 2026
@dustman9000

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 14, 2026
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Sep 14, 2026
@joshbranham

Copy link
Copy Markdown
Contributor Author

/label tide/merge-method-squash

@openshift-ci openshift-ci Bot added the tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. label Sep 14, 2026
@joshbranham

Copy link
Copy Markdown
Contributor Author

/retest

1 similar comment
@joshbranham

Copy link
Copy Markdown
Contributor Author

/retest

@geowa4 geowa4 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@AlexSmithGH

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 15, 2026
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: AlexSmithGH, dustman9000, geowa4, joshbranham

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [AlexSmithGH,dustman9000,geowa4,joshbranham]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@joshbranham

Copy link
Copy Markdown
Contributor Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 15, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit ccfdd26 into openshift:master Sep 15, 2026
5 checks passed
@joshbranham
joshbranham deleted the auto-release-pipeline branch September 15, 2026 14:52
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants