Skip to content

main-fix - #689

Closed
anilcanboga wants to merge 3 commits into
mainfrom
main-fix
Closed

anilcanboga wants to merge 3 commits into
mainfrom
main-fix

Conversation

@anilcanboga

Copy link
Copy Markdown
Collaborator

No description provided.

@vercel

vercel Bot commented Mar 9, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
tr-legacy-reactjs-org Ignored Ignored Preview Mar 9, 2026 6:26pm

Request Review

This branch had an error being deployed

1 failed deployment
Preview – tr-react-dev fc945eeb Deployed Mar 9, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant