DFBUGS-6106:[release-4.20] godeps: update grpc-go version - #792
Conversation
Signed-off-by: Nitin Goyal <nigoyal@redhat.com>
Signed-off-by: Nitin Goyal <nigoyal@redhat.com>
|
@iamniting: This pull request references [Jira Issue DFBUGS-6106](https://redhat.atlassian.net/browse/DFBUGS-6106), which is invalid:
Comment DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/test odf-operator-e2e-aws |
|
/override ci/prow/odf-operator-e2e-aws |
|
@iamniting: Overrode contexts on behalf of iamniting: ci/prow/odf-operator-e2e-aws DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/jira refresh |
|
@malayparida2000: This pull request references [Jira Issue DFBUGS-6106](https://redhat.atlassian.net/browse/DFBUGS-6106), which is valid. 3 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (sdurgbun@redhat.com), skipping review request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: iamniting, malayparida2000 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/override ci/prow/odf-operator-e2e-aws |
|
@malayparida2000: Overrode contexts on behalf of malayparida2000: ci/prow/odf-operator-e2e-aws DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
3581b4d
into
red-hat-storage:release-4.20
|
@iamniting: [Jira Issue DFBUGS-6106](https://redhat.atlassian.net/browse/DFBUGS-6106): All pull requests linked via external trackers have merged: [Jira Issue DFBUGS-6106](https://redhat.atlassian.net/browse/DFBUGS-6106) has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
Signed-off-by: Nitin Goyal nigoyal@redhat.com