Skip to content

fix: export cjs type declarations in package.json - #120

Merged
shahradelahi merged 1 commit into
mainfrom
fix/cjs-type-declarations
Aug 3, 2026
Merged

shahradelahi merged 1 commit into
mainfrom
fix/cjs-type-declarations

Conversation

@shahradelahi

Copy link
Copy Markdown
Owner

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 3, 2026

Copy link
Copy Markdown

馃 Changeset detected

Latest commit: fe90303

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@litehex/node-vault Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@shahradelahi
shahradelahi merged commit 011fb5d into main Aug 3, 2026
4 checks passed
@shahradelahi
shahradelahi deleted the fix/cjs-type-declarations branch August 3, 2026 17:57
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant