Skip to content

Bump @carbon/react from 1.115.0 to 1.116.0 - #5233

Merged
tekton-robot merged 2 commits into
mainfrom
dependabot/npm_and_yarn/carbon/react-1.116.0
Sep 14, 2026
Merged

tekton-robot merged 2 commits into
mainfrom
dependabot/npm_and_yarn/carbon/react-1.116.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 13, 2026

Copy link
Copy Markdown
Contributor

Bumps @carbon/react from 1.115.0 to 1.116.0.

Commits

@dependabot dependabot Bot added area/dependency Issues or PRs related to dependency changes dependencies Used by dependabot - identifies all PRs created by dependabot javascript Pull requests that update Javascript code kind/misc Categorizes issue or PR as a miscellaneuous one. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. labels Sep 13, 2026
@tekton-robot tekton-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 13, 2026
Bumps [@carbon/react](https://github.com/carbon-design-system/carbon/tree/HEAD/packages/carbon-react) from 1.115.0 to 1.116.0.
- [Release notes](https://github.com/carbon-design-system/carbon/releases)
- [Commits](https://github.com/carbon-design-system/carbon/commits/HEAD/packages/carbon-react)

---
updated-dependencies:
- dependency-name: "@carbon/react"
  dependency-version: 1.116.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/carbon/react-1.116.0 branch from f66bdf8 to c2ebd02 Compare September 13, 2026 23:25
@AlanGreene

Copy link
Copy Markdown
Member

Carbon now blocks using interactive content in label-like props, so our tooltips on text input / combo box / dropdown / etc. labels cause unit tests to fail.

A new approach will be needed before updating Carbon, ideally preserving or staying close to the current visual design, but resolving the accessibility issues caused by the current DOM structure.

Carbon 1.116.0 blocks use of interactive elements in label-like props.

Use the `decorator` prop to move the tooltip inside the visual wrapper
of the input element while avoiding the accessibility issues caused by
the previous resulting DOM that resulted in nested interactive elements.
@tekton-robot tekton-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 13, 2026
@AlanGreene

AlanGreene commented Sep 14, 2026

Copy link
Copy Markdown
Member

Updated to use the decorator prop on form elements to move the tooltip out of the label while keeping it inside the visual wrapper of the form element, avoiding nested interactive elements and the associated a11y issues. decorator is normally used for the AI label component which contains interactive elements so this is a permitted use.

This approach also avoids the need to mess with tab order to maintain the previous visual style.

Before:

image

After:

image

@AlanGreene AlanGreene left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Approving the Carbon update. However, as I've made visual changes to account for a breaking change in this release I'll hold off for another maintainer's review.

@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 14, 2026
@kaushalnavneet

Copy link
Copy Markdown
Contributor

Thanks, @AlanGreene. Looks good to me. I took a look at it locally, and the overall UX and tab order seem fine.
I was also curious about how the dropdown behaves when clicking the tooltip icon inside it. Hovering/clicking the icon shows the tooltip message as expected without opening the dropdown list, which is good.
Tests are passing, so I guess they're not impacted by this change.

@kaushalnavneet kaushalnavneet left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@tekton-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: AlanGreene, kaushalnavneet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [AlanGreene,kaushalnavneet]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 14, 2026
@tekton-robot
tekton-robot merged commit 5ede583 into main Sep 14, 2026
22 of 25 checks passed
@dependabot
dependabot Bot deleted the dependabot/npm_and_yarn/carbon/react-1.116.0 branch September 14, 2026 15:38
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/dependency Issues or PRs related to dependency changes dependencies Used by dependabot - identifies all PRs created by dependabot javascript Pull requests that update Javascript code kind/misc Categorizes issue or PR as a miscellaneuous one. lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants